Skip to content

fix(deps-dev): bump axepress/wp-graphql-stubs from 2.20.0 to 2.22.2 in /plugins/wp-graphql-smart-cache in the wp-graphql-smart-cache-composer-dev-minor-patch group - #4322

Open
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/composer/plugins/wp-graphql-smart-cache/wp-graphql-smart-cache-composer-dev-minor-patch-37d116d132
Open

fix(deps-dev): bump axepress/wp-graphql-stubs from 2.20.0 to 2.22.2 in /plugins/wp-graphql-smart-cache in the wp-graphql-smart-cache-composer-dev-minor-patch group#4322
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/composer/plugins/wp-graphql-smart-cache/wp-graphql-smart-cache-composer-dev-minor-patch-37d116d132

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 14, 2026

Copy link
Copy Markdown
Contributor

Bumps the wp-graphql-smart-cache-composer-dev-minor-patch group in /plugins/wp-graphql-smart-cache with 1 update: axepress/wp-graphql-stubs.

Updates axepress/wp-graphql-stubs from 2.20.0 to 2.22.2

Commits
  • 33d4311 Generate stubs for WPGraphQL v2.22.2
  • 812048f Generate stubs for WPGraphQL v2.22.1
  • 7dc624c Generate stubs for WPGraphQL v2.22.0
  • 0f90543 Generate stubs for WPGraphQL v2.21.1
  • 8b64d69 Generate stubs for WPGraphQL v2.21.0
  • See full diff in compare view

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore <dependency name> major version will close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)
  • @dependabot ignore <dependency name> minor version will close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)
  • @dependabot ignore <dependency name> will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)
  • @dependabot unignore <dependency name> will remove all of the ignore conditions of the specified dependency
  • @dependabot unignore <dependency name> <ignore condition> will remove the ignore condition of the specified dependency and ignore conditions

Bumps the wp-graphql-smart-cache-composer-dev-minor-patch group in /plugins/wp-graphql-smart-cache with 1 update: [axepress/wp-graphql-stubs](https://github.com/AxeWP/wp-graphql-stubs).


Updates `axepress/wp-graphql-stubs` from 2.20.0 to 2.22.2
- [Commits](AxeWP/wp-graphql-stubs@v2.20.0...v2.22.2)

---
updated-dependencies:
- dependency-name: axepress/wp-graphql-stubs
  dependency-version: 2.22.2
  dependency-type: direct:development
  update-type: version-update:semver-minor
  dependency-group: wp-graphql-smart-cache-composer-dev-minor-patch
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file php Pull requests that update Php code labels Sep 14, 2026
@dependabot
dependabot Bot requested a review from jasonbahl as a code owner September 14, 2026 01:25
@dependabot dependabot Bot added php Pull requests that update Php code dependencies Pull requests that update a dependency file labels Sep 14, 2026
@vercel

vercel Bot commented Sep 14, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
wpgraphql-com Skipped Skipped Sep 14, 2026 1:25am UTC

Request Review

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file php Pull requests that update Php code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants