Skip to content

ROX-17507: Automatically find start-release git SHA (#101) - #7302

Merged
vikin91 merged 1 commit into
masterfrom
ROX-17507--Automatically-find-start-release-git-SHA---101-
Aug 8, 2023
Merged

vikin91 merged 1 commit into
masterfrom
ROX-17507--Automatically-find-start-release-git-SHA---101-

Conversation

@ghost

@ghost ghost commented Aug 7, 2023

Copy link
Copy Markdown

Auto-generated PR to update the upstream GHA automation

@ghost ghost assigned vikin91 Aug 7, 2023
@ghost

ghost commented Aug 7, 2023

Copy link
Copy Markdown
Author

Images are ready for the commit at 7289c49.

To use with deploy scripts, first export MAIN_IMAGE_TAG=4.1.x-680-g7289c49fa4.

@openshift-ci

openshift-ci Bot commented Aug 7, 2023

Copy link
Copy Markdown

@Roxbot: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/gke-nongroovy-e2e-tests 7289c49 link false /test gke-nongroovy-e2e-tests
ci/prow/gke-qa-e2e-tests 7289c49 link false /test gke-qa-e2e-tests

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@vikin91
vikin91 merged commit 38c4131 into master Aug 8, 2023
@vikin91
vikin91 deleted the ROX-17507--Automatically-find-start-release-git-SHA---101- branch August 8, 2023 05:41
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant