Skip to content

[release-v1.34] Push releases additionally with a tag without the patch part of the version - #1571

Merged
ske-prow[bot] merged 2 commits into
stackitcloud:release-v1.34from
stackit-ske:cherry-pick-1570-to-release-v1.34
Aug 27, 2026
Merged

ske-prow[bot] merged 2 commits into
stackitcloud:release-v1.34from
stackit-ske:cherry-pick-1570-to-release-v1.34

Conversation

@stackit-ske

Copy link
Copy Markdown

This is an automated cherry-pick of #1570

/assign dergeberl

@stackit-ske stackit-ske added the kind/cherry-pick This is a cherry-pick and should be documented as such label Aug 27, 2026
@ske-prow ske-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Aug 27, 2026

@dergeberl dergeberl left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/kind enhancement

@ske-prow ske-prow Bot added kind/enhancement Enhancement, improvement, extension lgtm Indicates that a PR is ready to be merged. labels Aug 27, 2026
@ske-prow

ske-prow Bot commented Aug 27, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dergeberl

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow ske-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 27, 2026
@ske-prow

ske-prow Bot commented Aug 27, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: 89ad41c217309870936f4a734d4c0a1d9de84ef0

@ske-prow
ske-prow Bot merged commit 5b1300c into stackitcloud:release-v1.34 Aug 27, 2026
2 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. kind/cherry-pick This is a cherry-pick and should be documented as such kind/enhancement Enhancement, improvement, extension lgtm Indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants