Push releases additionally with a tag without the patch part of the version - #1570
Conversation
|
/cc @stackitcloud/ske-infrastructure |
6232e46 to
2297ebb
Compare
2297ebb to
afd616a
Compare
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: xoxys The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
LGTM label has been added. DetailsGit tree hash: 4b294c98c46ca7ca5000919ada84bf612fbaa84e |
|
@dergeberl: new pull request created: #1571 DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
@dergeberl: new pull request created: #1572 DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
@dergeberl: new pull request created: #1573 DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
How to categorize this PR?
/kind enhancement
/cherry-pick release-v1.34
/cherry-pick release-v1.35
/cherry-pick release-v1.36
What this PR does / why we need it:
This adds an additionally tag without the patch part of the semver.
Additionally it overwrites the tags instead of append, in an release there is no branch (empty), we don't need a version with v- and the actual version was 2 times in the list.
Which issue(s) this PR fixes:
Part of #1162
Special notes for your reviewer:
To test you can checkout different tags/branches and run
make imagesto check the tags.