Skip to content

add cherry pick to pull request template - #1534

Merged
ske-prow[bot] merged 2 commits into
mainfrom
update-pr-template-cherry-pick
Aug 24, 2026
Merged

ske-prow[bot] merged 2 commits into
mainfrom
update-pr-template-cherry-pick

Conversation

@breuerfelix

Copy link
Copy Markdown
Member

How to categorize this PR?

/label task

What this PR does / why we need it:
To not forget to cherry pick changes into the release branches, this PR adds them to the pull request template since almost all PRs should be cherry picked (except renovate PRs).

Which issue(s) this PR fixes:
Fixes #

Special notes for your reviewer:

Breaking changes:

/cherry-pick release-v1.34
/cherry-pick release-v1.35
/cherry-pick release-v1.36

Signed-off-by: Felix Breuer <f.breuer94@gmail.com>
@ske-prow

ske-prow Bot commented Aug 21, 2026

Copy link
Copy Markdown

@breuerfelix: The label(s) `/label task

cannot be applied. These labels are supported:tide/merge-method-merge, tide/merge-method-rebase, tide/merge-method-squash, ondemand/skip-deletion, breaking, no-release-note, rrwg-change, skip-review. Is this label configured under labels -> additional_labelsorlabels -> restricted_labelsinplugin.yaml`?

Details

In response to this:

How to categorize this PR?

/label task

What this PR does / why we need it:
To not forget to cherry pick changes into the release branches, this PR adds them to the pull request template since almost all PRs should be cherry picked (except renovate PRs).

Which issue(s) this PR fixes:
Fixes #

Special notes for your reviewer:

Breaking changes:

/cherry-pick release-v1.34
/cherry-pick release-v1.35
/cherry-pick release-v1.36

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@ske-prow ske-prow Bot added the do-not-merge/needs-kind Indicates a PR lacks a `kind/foo` label and requires one. label Aug 21, 2026
@breuerfelix
breuerfelix requested a review from a team August 21, 2026 15:12
@ske-prow ske-prow Bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Aug 21, 2026
@breuerfelix

Copy link
Copy Markdown
Member Author

/kind task

@ske-prow ske-prow Bot added kind/task General task and removed do-not-merge/needs-kind Indicates a PR lacks a `kind/foo` label and requires one. labels Aug 21, 2026
Comment thread .github/pull_request_template.md Outdated
Signed-off-by: Felix Breuer <f.breuer94@gmail.com>
@breuerfelix
breuerfelix requested a review from dergeberl August 24, 2026 07:58
@ske-prow ske-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Aug 24, 2026
@ske-prow

ske-prow Bot commented Aug 24, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dergeberl

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow ske-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 24, 2026
@ske-prow

ske-prow Bot commented Aug 24, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: 629a721193239b7a9645a9b2984d2c9d8585288c

@ske-prow
ske-prow Bot merged commit da6ac01 into main Aug 24, 2026
3 checks passed
@ske-prow
ske-prow Bot deleted the update-pr-template-cherry-pick branch August 24, 2026 09:19
@stackit-ske

Copy link
Copy Markdown

@breuerfelix: #1534 failed to apply on top of branch "release-v1.35":

Applying: add cherry pick to pull request template
.git/rebase-apply/patch:18: new blank line at EOF.
+
warning: 1 line adds whitespace errors.
Using index info to reconstruct a base tree...
M	.github/pull_request_template.md
M	docs/release-procedure.md
Falling back to patching base and 3-way merge...
Auto-merging .github/pull_request_template.md
Auto-merging docs/release-procedure.md
CONFLICT (content): Merge conflict in docs/release-procedure.md
error: Failed to merge in the changes.
hint: Use 'git am --show-current-patch=diff' to see the failed patch
hint: When you have resolved this problem, run "git am --continue".
hint: If you prefer to skip this patch, run "git am --skip" instead.
hint: To restore the original branch and stop patching, run "git am --abort".
hint: Disable this message with "git config set advice.mergeConflict false"
Patch failed at 0001 add cherry pick to pull request template

Details

In response to this:

How to categorize this PR?

/label task

What this PR does / why we need it:
To not forget to cherry pick changes into the release branches, this PR adds them to the pull request template since almost all PRs should be cherry picked (except renovate PRs).

Which issue(s) this PR fixes:
Fixes #

Special notes for your reviewer:

Breaking changes:

/cherry-pick release-v1.34
/cherry-pick release-v1.35
/cherry-pick release-v1.36

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@stackit-ske

Copy link
Copy Markdown

@breuerfelix: new pull request created: #1561

Details

In response to this:

How to categorize this PR?

/label task

What this PR does / why we need it:
To not forget to cherry pick changes into the release branches, this PR adds them to the pull request template since almost all PRs should be cherry picked (except renovate PRs).

Which issue(s) this PR fixes:
Fixes #

Special notes for your reviewer:

Breaking changes:

/cherry-pick release-v1.34
/cherry-pick release-v1.35
/cherry-pick release-v1.36

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@stackit-ske

Copy link
Copy Markdown

@breuerfelix: #1534 failed to apply on top of branch "release-v1.34":

Applying: add cherry pick to pull request template
.git/rebase-apply/patch:18: new blank line at EOF.
+
warning: 1 line adds whitespace errors.
Using index info to reconstruct a base tree...
M	.github/pull_request_template.md
M	docs/release-procedure.md
Falling back to patching base and 3-way merge...
Auto-merging .github/pull_request_template.md
Auto-merging docs/release-procedure.md
CONFLICT (content): Merge conflict in docs/release-procedure.md
error: Failed to merge in the changes.
hint: Use 'git am --show-current-patch=diff' to see the failed patch
hint: When you have resolved this problem, run "git am --continue".
hint: If you prefer to skip this patch, run "git am --skip" instead.
hint: To restore the original branch and stop patching, run "git am --abort".
hint: Disable this message with "git config set advice.mergeConflict false"
Patch failed at 0001 add cherry pick to pull request template

Details

In response to this:

How to categorize this PR?

/label task

What this PR does / why we need it:
To not forget to cherry pick changes into the release branches, this PR adds them to the pull request template since almost all PRs should be cherry picked (except renovate PRs).

Which issue(s) this PR fixes:
Fixes #

Special notes for your reviewer:

Breaking changes:

/cherry-pick release-v1.34
/cherry-pick release-v1.35
/cherry-pick release-v1.36

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

ske-prow Bot pushed a commit that referenced this pull request Aug 24, 2026
* add cherry pick to pull request template



* add comment for cherry picks



---------

Signed-off-by: Felix Breuer <f.breuer94@gmail.com>
ske-prow Bot pushed a commit that referenced this pull request Aug 24, 2026
* add cherry pick to pull request template



* add comment for cherry picks



---------

Signed-off-by: Felix Breuer <f.breuer94@gmail.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. kind/task General task lgtm Indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants