Skip to content

Update kubernetes monorepo to v0.34.11 (release-v1.34) - #1521

Merged
ske-prow[bot] merged 1 commit into
release-v1.34from
renovate/release-v1.34-kubernetes-monorepo
Aug 21, 2026
Merged

ske-prow[bot] merged 1 commit into
release-v1.34from
renovate/release-v1.34-kubernetes-monorepo

Conversation

@ske-renovate-operator

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
k8s.io/apimachinery v0.34.10v0.34.11 age confidence
k8s.io/mount-utils v0.34.10v0.34.11 age confidence

Release Notes

kubernetes/apimachinery (k8s.io/apimachinery)

v0.34.11

Compare Source

kubernetes/mount-utils (k8s.io/mount-utils)

v0.34.11

Compare Source


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about these updates again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate CLI.

@ske-renovate-operator ske-renovate-operator Bot added kind/upgrade Upgrading components or dependencies (e.g., by renovate) skip-review Indicates a PR is trusted, used by tide for auto-merging PRs. labels Aug 21, 2026
@ske-renovate-operator
ske-renovate-operator Bot requested a review from a team August 21, 2026 04:33
@ske-renovate-operator ske-renovate-operator Bot added the kind/upgrade Upgrading components or dependencies (e.g., by renovate) label Aug 21, 2026
@ske-prow

ske-prow Bot commented Aug 21, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign juliusste for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow ske-prow Bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Aug 21, 2026
@ske-prow
ske-prow Bot merged commit af48c41 into release-v1.34 Aug 21, 2026
4 checks passed
@ske-prow
ske-prow Bot deleted the renovate/release-v1.34-kubernetes-monorepo branch August 21, 2026 04:36
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/upgrade Upgrading components or dependencies (e.g., by renovate) size/S Denotes a PR that changes 10-29 lines, ignoring generated files. skip-review Indicates a PR is trusted, used by tide for auto-merging PRs.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants