Skip to content

Update dependency chainguard-dev/apko to v1.2.37 (release-v1.35) - #1513

Merged
ske-prow[bot] merged 1 commit into
release-v1.35from
renovate/release-v1.35-devtools-non-major
Aug 20, 2026
Merged

ske-prow[bot] merged 1 commit into
release-v1.35from
renovate/release-v1.35-devtools-non-major

Conversation

@ske-renovate-operator

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Update Change Pending
chainguard-dev/apko patch v1.2.36v1.2.37 v1.2.38

Release Notes

chainguard-dev/apko (chainguard-dev/apko)

v1.2.37

Compare Source

Changelog
  • 47ec388 build(deps): bump chainguard-dev/actions/setup-registry from 1.6.30 to 1.6.31 (#​2396)
  • e669022 build(deps): bump chainguard.dev/sdk from 0.1.170 to 0.1.184 (#​2394)
  • bf7cab9 build(deps): bump github.com/google/go-containerregistry from 0.21.8 to 0.21.9 (#​2391)
  • 7f1606c build(deps): bump github.com/klauspost/compress from 1.19.1 to 1.19.2 (#​2393)
  • 5bf3f6b build(deps): bump google.golang.org/api from 0.291.0 to 0.292.0 (#​2392)
  • c9d67cb certificates: also add kaniko certs (#​2395)

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate CLI.

@ske-renovate-operator ske-renovate-operator Bot added kind/upgrade Upgrading components or dependencies (e.g., by renovate) skip-review Indicates a PR is trusted, used by tide for auto-merging PRs. labels Aug 20, 2026
@ske-renovate-operator

Copy link
Copy Markdown
Contributor Author

/label skip-review

@ske-prow

ske-prow Bot commented Aug 20, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign breuerfelix for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow ske-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Aug 20, 2026
@ske-prow
ske-prow Bot merged commit a38e7ce into release-v1.35 Aug 20, 2026
4 checks passed
@ske-prow
ske-prow Bot deleted the renovate/release-v1.35-devtools-non-major branch August 20, 2026 00:35
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/upgrade Upgrading components or dependencies (e.g., by renovate) size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. skip-review Indicates a PR is trusted, used by tide for auto-merging PRs.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants