Skip to content

Update dependency chainguard-dev/apko to v1.2.27 (release-v1.34) - #1392

Merged
ske-prow[bot] merged 1 commit into
release-v1.34from
renovate/release-v1.34-devtools-non-major
Jul 20, 2026
Merged

ske-prow[bot] merged 1 commit into
release-v1.34from
renovate/release-v1.34-devtools-non-major

Conversation

@ske-renovate-operator

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Update Change
chainguard-dev/apko patch v1.2.26v1.2.27

Release Notes

chainguard-dev/apko (chainguard-dev/apko)

v1.2.27

Compare Source

Changelog
  • 63234ec build(deps): bump chainguard.dev/sdk from 0.1.119 to 0.1.127 (#​2330)

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate.

@ske-renovate-operator ske-renovate-operator Bot added kind/upgrade Upgrading components or dependencies (e.g., by renovate) skip-review Indicates a PR is trusted, used by tide for auto-merging PRs. labels Jul 20, 2026
@ske-renovate-operator

Copy link
Copy Markdown
Contributor Author

/label skip-review

@ske-prow

ske-prow Bot commented Jul 20, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign kumm-kai for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow ske-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Jul 20, 2026
@ske-prow
ske-prow Bot merged commit 0f66d6d into release-v1.34 Jul 20, 2026
3 checks passed
@ske-prow
ske-prow Bot deleted the renovate/release-v1.34-devtools-non-major branch July 20, 2026 04:35
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/upgrade Upgrading components or dependencies (e.g., by renovate) size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. skip-review Indicates a PR is trusted, used by tide for auto-merging PRs.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants