Skip to content

Update dependency chainguard-dev/apko to v1.2.26 (release-v1.34) - #1388

Merged
ske-prow[bot] merged 1 commit into
release-v1.34from
renovate/release-v1.34-devtools-non-major
Jul 17, 2026
Merged

ske-prow[bot] merged 1 commit into
release-v1.34from
renovate/release-v1.34-devtools-non-major

Conversation

@ske-renovate-operator

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Update Change
chainguard-dev/apko patch v1.2.25v1.2.26

Release Notes

chainguard-dev/apko (chainguard-dev/apko)

v1.2.26

Compare Source

Changelog
  • 18cd01e build(deps): bump chainguard-dev/actions/setup-registry from 1.6.26 to 1.6.27 (#​2329)
  • 118c537 build(deps): bump github/codeql-action/analyze from 4.36.3 to 4.37.0 (#​2327)
  • e7f12d4 build(deps): bump github/codeql-action/init from 4.36.3 to 4.37.0 (#​2326)
  • aa9ca73 chore(release): add more arches (#​2242)

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate.

@ske-renovate-operator ske-renovate-operator Bot added kind/upgrade Upgrading components or dependencies (e.g., by renovate) skip-review Indicates a PR is trusted, used by tide for auto-merging PRs. labels Jul 17, 2026
@ske-renovate-operator

Copy link
Copy Markdown
Contributor Author

/label skip-review

@ske-renovate-operator ske-renovate-operator Bot added the kind/upgrade Upgrading components or dependencies (e.g., by renovate) label Jul 17, 2026
@ske-prow

ske-prow Bot commented Jul 17, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign robinschneider for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow ske-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Jul 17, 2026
@ske-prow
ske-prow Bot merged commit 4d31fe0 into release-v1.34 Jul 17, 2026
3 checks passed
@ske-prow
ske-prow Bot deleted the renovate/release-v1.34-devtools-non-major branch July 17, 2026 16:35
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/upgrade Upgrading components or dependencies (e.g., by renovate) size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. skip-review Indicates a PR is trusted, used by tide for auto-merging PRs.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants