Skip to content

Update module google.golang.org/grpc to v1.82.1 (release-v1.34) - #1374

Merged
ske-prow[bot] merged 1 commit into
release-v1.34from
renovate/release-v1.34-google.golang.org-grpc-1.x
Jul 16, 2026
Merged

ske-prow[bot] merged 1 commit into
release-v1.34from
renovate/release-v1.34-google.golang.org-grpc-1.x

Conversation

@ske-renovate-operator

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
google.golang.org/grpc v1.82.0v1.82.1 age confidence

Release Notes

grpc/grpc-go (google.golang.org/grpc)

v1.82.1: Release 1.82.1

Compare Source

Security

  • server: Stop reading from the connection when flooded by HTTP/2 frames. The default value for this limit is 100 frames, excluding DATA and HEADERS, and may be changed by setting environment variable GRPC_GO_EXPERIMENTAL_CONTROL_BUFFER_THROTTLE_LIMIT.
  • xds/rbac: Support Metadata and RequestedServerName permissions matcher fields. If present in a DENY rule, previously these would be ignored and fail-open.
  • xds/rbac: Fix panic when parsing unsupported fields in NotRule/NotId permissions.
  • xds/rbac: Support the deprecated source_ip principal identifier by treating it as equivalent to direct_remote_ip.

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate.

@ske-renovate-operator ske-renovate-operator Bot added kind/upgrade Upgrading components or dependencies (e.g., by renovate) skip-review Indicates a PR is trusted, used by tide for auto-merging PRs. labels Jul 16, 2026
@ske-renovate-operator

Copy link
Copy Markdown
Contributor Author

/label skip-review

@ske-prow

ske-prow Bot commented Jul 16, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign timebertt for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow ske-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Jul 16, 2026
@ske-prow
ske-prow Bot merged commit a340bda into release-v1.34 Jul 16, 2026
3 checks passed
@ske-prow
ske-prow Bot deleted the renovate/release-v1.34-google.golang.org-grpc-1.x branch July 16, 2026 00:37
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/upgrade Upgrading components or dependencies (e.g., by renovate) size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. skip-review Indicates a PR is trusted, used by tide for auto-merging PRs.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants