Skip to content

[release-v1.35] node-controller: support multi-nic - #1343

Merged
ske-prow[bot] merged 7 commits into
stackitcloud:release-v1.35from
stackit-ske:cherry-pick-1307-to-release-v1.35
Jul 8, 2026
Merged

ske-prow[bot] merged 7 commits into
stackitcloud:release-v1.35from
stackit-ske:cherry-pick-1307-to-release-v1.35

Conversation

@stackit-ske

Copy link
Copy Markdown

This is an automated cherry-pick of #1307

/assign nschad

hown3d and others added 7 commits July 8, 2026 13:50
Signed-off-by: Lukas Hoehl <lukas.hoehl@stackit.cloud>
Signed-off-by: Lukas Hoehl <lukas.hoehl@stackit.cloud>
Signed-off-by: Lukas Hoehl <lukas.hoehl@stackit.cloud>
Signed-off-by: Lukas Hoehl <lukas.hoehl@stackit.cloud>
Co-authored-by: Niclas Schad <niclas.schad@stackit.cloud>
Signed-off-by: Lukas Hoehl <lukas.hoehl@stackit.cloud>
Signed-off-by: Lukas Hoehl <lukas.hoehl@stackit.cloud>
@stackit-ske stackit-ske added the kind/cherry-pick This is a cherry-pick and should be documented as such label Jul 8, 2026
@ske-prow ske-prow Bot assigned nschad Jul 8, 2026
@ske-prow ske-prow Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Jul 8, 2026
@ske-prow

ske-prow Bot commented Jul 8, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: nschad

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow ske-prow Bot added lgtm Indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Jul 8, 2026
@ske-prow

ske-prow Bot commented Jul 8, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: cb58fd3a5324c49b0008f08bb156365fd65f914d

@ske-prow
ske-prow Bot merged commit 2d9533d into stackitcloud:release-v1.35 Jul 8, 2026
2 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. kind/cherry-pick This is a cherry-pick and should be documented as such lgtm Indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants