Skip to content

Update devTools non-major (release-v1.35) - #1273

Merged
ske-prow[bot] merged 1 commit into
release-v1.35from
renovate/release-v1.35-devtools-non-major
Jun 26, 2026
Merged

ske-prow[bot] merged 1 commit into
release-v1.35from
renovate/release-v1.35-devtools-non-major

Conversation

@ske-renovate-operator

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Update Change
ko-build/ko minor v0.18.1v0.19.0

Release Notes

ko-build/ko (ko-build/ko)

v0.19.0

Compare Source

What's Changed
New Contributors

Full Changelog: ko-build/ko@v0.18.1...v0.19.0


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate.

@ske-renovate-operator ske-renovate-operator Bot added kind/upgrade Upgrading components or dependencies (e.g., by renovate) skip-review Indicates a PR is trusted, used by tide for auto-merging PRs. labels Jun 26, 2026
@ske-renovate-operator

Copy link
Copy Markdown
Contributor Author

/label skip-review

@ske-prow

ske-prow Bot commented Jun 26, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign dergeberl for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow ske-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Jun 26, 2026
@ske-prow
ske-prow Bot merged commit 6718ced into release-v1.35 Jun 26, 2026
3 checks passed
@ske-prow
ske-prow Bot deleted the renovate/release-v1.35-devtools-non-major branch June 26, 2026 16:37
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/upgrade Upgrading components or dependencies (e.g., by renovate) size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. skip-review Indicates a PR is trusted, used by tide for auto-merging PRs.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants