chore: add renovate config to allow major version updates for GH actions [skip-build] - #5314
chore: add renovate config to allow major version updates for GH actions [skip-build]#5314Zaperex wants to merge 2 commits into
Conversation
…tions Signed-off-by: Frank Kong <frkong@redhat.com> rh-pre-commit.version: 2.4.0 rh-pre-commit.check-secrets: ENABLED
|
/override "ci/prow/e2e-ocp-helm" |
|
@nickboldt: Overrode contexts on behalf of nickboldt: Build Image, Build with Node.js, Test with Node.js, ci/prow/e2e-ocp-helm DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #5314 +/- ##
==========================================
- Coverage 63.68% 59.69% -3.99%
==========================================
Files 121 109 -12
Lines 2288 2062 -226
Branches 503 501 -2
==========================================
- Hits 1457 1231 -226
+ Misses 830 829 -1
- Partials 1 2 +1
Continue to review full report in Codecov by Harness.
🚀 New features to boost your workflow:
|
|
/retest |
|
This PR is stale because it has been open 7 days with no activity. Remove stale label or comment or this will be closed in 21 days. |
|
|
@Zaperex: The following test failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |



Description
Updates the renovate config to allow renovate to update GH actions to the latest major version
Which issue(s) does this PR fix
PR acceptance criteria
Please make sure that the following steps are complete:
How to test changes / Special notes to the reviewer