balance, proxy: fix TiDB CPU imbalance when balance.policy="resource" (#1173) - #1175
Merged
ti-chi-bot[bot] merged 3 commits intoJun 15, 2026
Conversation
8 tasks
Codecov Report❌ Patch coverage is Additional details and impacted files@@ Coverage Diff @@
## release-nextgen-202603 #1175 +/- ##
=========================================================
Coverage ? 67.55%
=========================================================
Files ? 149
Lines ? 16208
Branches ? 0
=========================================================
Hits ? 10949
Misses ? 4523
Partials ? 736
Flags with carried forward coverage won't be shown. Click here to find out more. ☔ View full report in Codecov by Harness. 🚀 New features to boost your workflow:
|
djshow832
approved these changes
Jun 15, 2026
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: djshow832 The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
[LGTM Timeline notifier]Timeline:
|
ti-chi-bot
Bot
merged commit Jun 15, 2026
8ad2a00
into
pingcap:release-nextgen-202603
6 checks passed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This is an automated cherry-pick of #1173
What problem does this PR solve?
Issue Number: close #1172
Problem Summary:
ScoreBasedRoute.updateGroups().What is changed and how it works:
physicalOwnerandscoreOwnerto thebackendWrapperso that it can always find the right backend.redirectingBackendIDbecause it's unnecessary now.BackendConnManager.redirectResCh.BackendConnManager.tryRedirect.Check List
Tests
Notable changes
Release note
Please refer to Release Notes Language Style Guide to write a quality release note.