Skip to content

fix mac address src for lb ip in arp response when openelb pod restart on another k8s node - #360

Open
wenwenxiong wants to merge 4 commits into
openelb:masterfrom
wenwenxiong:master
Open

fix mac address src for lb ip in arp response when openelb pod restart on another k8s node#360
wenwenxiong wants to merge 4 commits into
openelb:masterfrom
wenwenxiong:master

Conversation

@wenwenxiong

Copy link
Copy Markdown

Description

What type of PR is this ?:
the first openelb deploy will generate arp item in switch, for example 172.31.249.200 lb ip
01

when openelb pod restart on another k8s node(fisrt in BAGG52 peer node, restart on XGE2/0/17 peer node)
the arp table source endpoint (from BAGG52 to XGE2/0/17) in switch will change
02

if wen did not update lb ip mac addresses like in this pr.
the will lead network unreach for switch gateway to lb ip

ping -vpn-instance N6.1001-MEC 172.31.249.200

because the source endpoint not consistency

Related links:

@ks-ci-bot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: wenwenxiong
To complete the pull request process, please assign feynmanzhou after the PR has been reviewed.
You can assign the PR to them by writing /assign @feynmanzhou in a comment when ready.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ks-ci-bot ks-ci-bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 22, 2023
@wenwenxiong

Copy link
Copy Markdown
Author

/assign @FeynmanZhou

@renyunkang

Copy link
Copy Markdown
Member

It is not recommended that you modify it this way. This breaks the working principle of openelb layer2 mode. Your modification will refresh all existing allocations to the mac of the node where openelb-manager is located instead of the nodes available in the cluster.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants