Skip to content
This repository was archived by the owner on Mar 23, 2026. It is now read-only.

Update CODEOWNERS - #13523

Merged
alexrashed merged 1 commit into
mainfrom
codeowners-auto-updates
Dec 15, 2025
Merged

alexrashed merged 1 commit into
mainfrom
codeowners-auto-updates

Conversation

@localstack-bot

Copy link
Copy Markdown
Contributor

Automatic update of CODEOWNERS file.

@localstack-bot localstack-bot added docs: skip Pull request does not require documentation changes notes: skip Pull request does not have to be mentioned in the release notes semver: patch Non-breaking changes which can be included in patch releases labels Dec 15, 2025

@alexrashed alexrashed left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Nice to see lots of new service owners from @k-a-il @carole-lavillonniere @tiurin @silv-io! 馃挴 馃ぉ 馃Ω馃徑

@alexrashed
alexrashed merged commit 5c50585 into main Dec 15, 2025
27 checks passed
@alexrashed
alexrashed deleted the codeowners-auto-updates branch December 15, 2025 08:30
viren-nadkarni pushed a commit that referenced this pull request Dec 15, 2025
Co-authored-by: LocalStack Bot <localstack-bot@users.noreply.github.com>
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

docs: skip Pull request does not require documentation changes notes: skip Pull request does not have to be mentioned in the release notes semver: patch Non-breaking changes which can be included in patch releases

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants