Skip to content
This repository was archived by the owner on Mar 23, 2026. It is now read-only.

IaC: AWS and CloudFormation features support catalog - #13026

Closed
k-a-il wants to merge 3 commits into
mainfrom
iac/catalog-skeleton
Closed

k-a-il wants to merge 3 commits into
mainfrom
iac/catalog-skeleton

Conversation

@k-a-il

@k-a-il k-a-il commented Aug 19, 2025

Copy link
Copy Markdown
Collaborator

Motivation

Create initial structure for AWS and CloudFormation catalogs.

Changes

  • Adds AwsServicesCatalog and CloudFormationCatalog
  • Logic will be added in next PRs

@k-a-il k-a-il self-assigned this Aug 19, 2025
@k-a-il k-a-il changed the title IaC: aws and cfn catalog skeletons IaC: AWS and CloudFormation features support catalog Aug 19, 2025
@k-a-il
k-a-il requested a review from bentsku August 19, 2025 09:29
@k-a-il k-a-il added the semver: patch Non-breaking changes which can be included in patch releases label Aug 19, 2025
@github-actions

github-actions Bot commented Aug 19, 2025

Copy link
Copy Markdown

Test Results - Preflight, Unit

22 139 tests  ±0   20 404 ✅ ±0   6m 34s ⏱️ +3s
     1 suites ±0    1 735 💤 ±0 
     1 files   ±0        0 ❌ ±0 

Results for commit 8b3c93d. ± Comparison against base commit 1baa022.

♻️ This comment has been updated with latest results.

@github-actions

github-actions Bot commented Aug 19, 2025

Copy link
Copy Markdown

Test Results (amd64) - Acceptance

7 tests  ±0   5 ✅ ±0   3m 9s ⏱️ -1s
1 suites ±0   2 💤 ±0 
1 files   ±0   0 ❌ ±0 

Results for commit 8b3c93d. ± Comparison against base commit 1baa022.

♻️ This comment has been updated with latest results.

@github-actions

github-actions Bot commented Aug 19, 2025

Copy link
Copy Markdown

Test Results (amd64) - Integration, Bootstrap

    5 files      5 suites   2h 21m 2s ⏱️
4 989 tests 4 397 ✅ 592 💤 0 ❌
4 995 runs  4 397 ✅ 598 💤 0 ❌

Results for commit 8b3c93d.

♻️ This comment has been updated with latest results.

@github-actions

github-actions Bot commented Aug 19, 2025

Copy link
Copy Markdown

LocalStack Community integration with Pro

    2 files  ±0      2 suites  ±0   1h 45m 9s ⏱️ + 2m 38s
4 630 tests ±0  4 190 ✅ ±0  440 💤 ±0  0 ❌ ±0 
4 632 runs  ±0  4 190 ✅ ±0  442 💤 ±0  0 ❌ ±0 

Results for commit 8b3c93d. ± Comparison against base commit 1baa022.

♻️ This comment has been updated with latest results.

@k-a-il
k-a-il requested review from a team, ArthurAkh, bentsku, sannya-singal and silv-io and removed request for ArthurAkh, bentsku, sannya-singal and silv-io August 19, 2025 12:12
@k-a-il
k-a-il marked this pull request as ready for review August 19, 2025 13:00

@bentsku bentsku left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

edit: oops, sent without the full message 😅

This is great, thanks for adding this! It will be easier to iterate over it collectively 🚀
I think it would be good to add a CODEOWNERS entry for this module back to our squad, so it'd be easier for review 👍

I only have a comment around the factory of the class, in order to extend it from Pro and allow us to transparently modify behavior of the callers of the class

Comment thread localstack-core/localstack/utils/catalog/aws_catalog.py
@k-a-il
k-a-il requested a review from a team August 19, 2025 15:26
@k-a-il

k-a-il commented Sep 3, 2025

Copy link
Copy Markdown
Collaborator Author

Closing this PR because it is part of #13027

@k-a-il k-a-il closed this Sep 3, 2025
@k-a-il
k-a-il deleted the iac/catalog-skeleton branch February 27, 2026 14:50
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

semver: patch Non-breaking changes which can be included in patch releases

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants