Skip to content

build(deps): bump the maven-dependencies group with 7 updates - #4906

Closed
dependabot[bot] wants to merge 2 commits into
masterfrom
dependabot/maven/maven-dependencies-e8b31a607b
Closed

dependabot[bot] wants to merge 2 commits into
masterfrom
dependabot/maven/maven-dependencies-e8b31a607b

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 14, 2026

Copy link
Copy Markdown
Contributor

Bumps the maven-dependencies group with 7 updates:

Package From To
software.amazon.awssdk:sts 2.54.13 2.54.16
software.amazon.awssdk:auth 2.54.13 2.54.16
software.amazon.awssdk:http-auth-aws 2.54.13 2.54.16
software.amazon.awssdk:http-auth-spi 2.54.13 2.54.16
software.amazon.awssdk:http-client-spi 2.54.13 2.54.16
software.amazon.awssdk:utils 2.54.13 2.54.16
org.bitbucket.b_c:jose4j 0.9.6 0.9.7

Updates software.amazon.awssdk:sts from 2.54.13 to 2.54.16

Updates software.amazon.awssdk:auth from 2.54.13 to 2.54.16

Updates software.amazon.awssdk:http-auth-aws from 2.54.13 to 2.54.16

Updates software.amazon.awssdk:http-auth-spi from 2.54.13 to 2.54.16

Updates software.amazon.awssdk:http-client-spi from 2.54.13 to 2.54.16

Updates software.amazon.awssdk:utils from 2.54.13 to 2.54.16

Updates software.amazon.awssdk:auth from 2.54.13 to 2.54.16

Updates software.amazon.awssdk:http-auth-aws from 2.54.13 to 2.54.16

Updates software.amazon.awssdk:http-auth-spi from 2.54.13 to 2.54.16

Updates software.amazon.awssdk:http-client-spi from 2.54.13 to 2.54.16

Updates software.amazon.awssdk:utils from 2.54.13 to 2.54.16

Updates org.bitbucket.b_c:jose4j from 0.9.6 to 0.9.7

Changelog

Sourced from [org.bitbucket.b_c:jose4j's changelog](https://bitbucket.org/b_c/jose4j/src/master/Release Notes.md).

jose4j-0.9.7 - September 9, 2026
  • update test dependencies
  • cap the size of numbers for better resource utilization in JSON processing (hat tip to Mike Read for the tip)
  • fix some typos and other issues in javadoc and log messages
  • updates for newer maven central publishing
Commits
  • 756257e [maven-release-plugin] prepare for next development iteration
  • 5db177b Look for Tag mismatch with no "!" b/c apparently maybe that changed
  • 71a6650 Merged in strict-aud-validation (pull request #26)
  • 05ce887 Compiler source and target to 1.8
  • f7c6da8 Update slf4j, bouncycastle, and mockito dependencies
  • 3ac5eb4 Merged in fix-comment-typos (pull request #27)
  • a87874c Fix stated default in Get.setResponseBodySizeLimit javadoc
  • 51bdce2 fix log message arg number missmatch
  • 7d8c470 Another log message arg number mismatch with a different fix
  • 908a55e cap the size of numbers for better resource utilization in JSON processing (h...
  • Additional commits viewable in compare view

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore <dependency name> major version will close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)
  • @dependabot ignore <dependency name> minor version will close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)
  • @dependabot ignore <dependency name> will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)
  • @dependabot unignore <dependency name> will remove all of the ignore conditions of the specified dependency
  • @dependabot unignore <dependency name> <ignore condition> will remove the ignore condition of the specified dependency and ignore conditions

Bumps the maven-dependencies group with 7 updates:

| Package | From | To |
| --- | --- | --- |
| software.amazon.awssdk:sts | `2.54.13` | `2.54.16` |
| software.amazon.awssdk:auth | `2.54.13` | `2.54.16` |
| software.amazon.awssdk:http-auth-aws | `2.54.13` | `2.54.16` |
| software.amazon.awssdk:http-auth-spi | `2.54.13` | `2.54.16` |
| software.amazon.awssdk:http-client-spi | `2.54.13` | `2.54.16` |
| software.amazon.awssdk:utils | `2.54.13` | `2.54.16` |
| [org.bitbucket.b_c:jose4j](https://bitbucket.org/b_c/jose4j) | `0.9.6` | `0.9.7` |


Updates `software.amazon.awssdk:sts` from 2.54.13 to 2.54.16

Updates `software.amazon.awssdk:auth` from 2.54.13 to 2.54.16

Updates `software.amazon.awssdk:http-auth-aws` from 2.54.13 to 2.54.16

Updates `software.amazon.awssdk:http-auth-spi` from 2.54.13 to 2.54.16

Updates `software.amazon.awssdk:http-client-spi` from 2.54.13 to 2.54.16

Updates `software.amazon.awssdk:utils` from 2.54.13 to 2.54.16

Updates `software.amazon.awssdk:auth` from 2.54.13 to 2.54.16

Updates `software.amazon.awssdk:http-auth-aws` from 2.54.13 to 2.54.16

Updates `software.amazon.awssdk:http-auth-spi` from 2.54.13 to 2.54.16

Updates `software.amazon.awssdk:http-client-spi` from 2.54.13 to 2.54.16

Updates `software.amazon.awssdk:utils` from 2.54.13 to 2.54.16

Updates `org.bitbucket.b_c:jose4j` from 0.9.6 to 0.9.7
- [Changelog](https://bitbucket.org/b_c/jose4j/src/master/Release Notes.md)
- [Commits](https://bitbucket.org/b_c/jose4j/branches/compare/jose4j-0.9.7..jose4j-0.9.6)

---
updated-dependencies:
- dependency-name: software.amazon.awssdk:sts
  dependency-version: 2.54.16
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: maven-dependencies
- dependency-name: software.amazon.awssdk:auth
  dependency-version: 2.54.16
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: maven-dependencies
- dependency-name: software.amazon.awssdk:http-auth-aws
  dependency-version: 2.54.16
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: maven-dependencies
- dependency-name: software.amazon.awssdk:http-auth-spi
  dependency-version: 2.54.16
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: maven-dependencies
- dependency-name: software.amazon.awssdk:http-client-spi
  dependency-version: 2.54.16
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: maven-dependencies
- dependency-name: software.amazon.awssdk:utils
  dependency-version: 2.54.16
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: maven-dependencies
- dependency-name: software.amazon.awssdk:auth
  dependency-version: 2.54.16
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: maven-dependencies
- dependency-name: software.amazon.awssdk:http-auth-aws
  dependency-version: 2.54.16
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: maven-dependencies
- dependency-name: software.amazon.awssdk:http-auth-spi
  dependency-version: 2.54.16
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: maven-dependencies
- dependency-name: software.amazon.awssdk:http-client-spi
  dependency-version: 2.54.16
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: maven-dependencies
- dependency-name: software.amazon.awssdk:utils
  dependency-version: 2.54.16
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: maven-dependencies
- dependency-name: org.bitbucket.b_c:jose4j
  dependency-version: 0.9.7
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: maven-dependencies
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file java Pull requests that update Java code labels Sep 14, 2026
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Sep 14, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: dependabot[bot]
Once this PR has been reviewed and has the lgtm label, please assign brendandburns for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. and removed size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Sep 14, 2026
@dependabot @github

dependabot Bot commented on behalf of github Sep 15, 2026

Copy link
Copy Markdown
Contributor Author

Looks like these dependencies are updatable in another way, so this is no longer needed.

@dependabot dependabot Bot closed this Sep 15, 2026
@dependabot
dependabot Bot deleted the dependabot/maven/maven-dependencies-e8b31a607b branch September 15, 2026 02:03
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. dependencies Pull requests that update a dependency file java Pull requests that update Java code size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants