README updates - #3220
Open
stephenfin wants to merge 4 commits into
Open
Conversation
Just point to OWNERS. Signed-off-by: Stephen Finucane <stephenfin@redhat.com>
The newest release on this list, v1.26.0, is 10 releases (> 3 years) old. Signed-off-by: Stephen Finucane <stephenfin@redhat.com>
This is redundant info: everyone can read the LICENSE file themselves. Signed-off-by: Stephen Finucane <stephenfin@redhat.com>
Signed-off-by: Stephen Finucane <stephenfin@redhat.com>
Contributor
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
mandre
reviewed
Sep 14, 2026
|
|
||
| Refer to [Getting Started Guide](/docs/developers-guide.md/) for setting up development environment and contributing. | ||
| Implements the Kubernetes cloud controller manager interface for OpenStack, managing nodes, routes, and load balancers. | ||
| [Docs](/docs/openstack-cloud-controller-manager/using-openstack-cloud-controller-manager.md/) |
Contributor
There was a problem hiding this comment.
I suggest adding a new line for consistency with other components, as it would otherwise render differently in markdown.
Suggested change
| [Docs](/docs/openstack-cloud-controller-manager/using-openstack-cloud-controller-manager.md/) | |
| [Docs](/docs/openstack-cloud-controller-manager/using-openstack-cloud-controller-manager.md/) |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
What this PR does / why we need it:
README: Remove out-of-date maintainers listREADME: Remove old release infoREADME: Remove license sectionREADME: Provide more information about each componentWhich issue this PR fixes(if applicable):
(none)
Special notes for reviewers:
(none)
Release note: