fix: resolve internal dep updates recursively - #2294
Open
trueberryless wants to merge 4 commits into
Open
trueberryless wants to merge 4 commits into
trueberryless wants to merge 4 commits into
Conversation
🦋 Changeset detectedLatest commit: 6c500d6 The changes in this PR will be included in the next version bump. This PR includes changesets to release 2 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
✅ Deploy Preview for changesets ready!
To edit notification comments on pull requests, go to your Netlify project configuration. |
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #2294 +/- ##
==========================================
+ Coverage 84.57% 84.65% +0.08%
==========================================
Files 80 80
Lines 3177 3194 +17
Branches 894 900 +6
==========================================
+ Hits 2687 2704 +17
Misses 433 433
Partials 57 57 ☔ View full report in Codecov by Harness. 🚀 New features to boost your workflow:
|
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Description
This PR implements recursive internal dependency resolution in the
apply-release-planpackage to be able to list commit references in case the chain looks like e.g.a -> b -> c.I added some tests with small functions to generate the dependency graph.
@changesets/changelog-githubgenerates changelog with missing commit refs #918