python3Packages.aodhclient: 3.10.1 -> 3.11.0, rename from python3Packages.python-aodhclient to match python metadata - #546717
Merged
Conversation
Member
Author
|
Vinetos
force-pushed
the
fix/aodh
branch
2 times, most recently
from
July 29, 2026 18:37
876f7d9 to
03421c0
Compare
Member
Author
|
SuperSandro2000
force-pushed
the
fix/aodh
branch
from
July 29, 2026 19:01
03421c0 to
d9484ff
Compare
Member
Author
|
Sigmanificient
previously requested changes
Jul 29, 2026
anthonyroussel
approved these changes
Jul 31, 2026
anthonyroussel
left a comment
Member
There was a problem hiding this comment.
nixpkgs-review result
Generated using nixpkgs-review.
Command: nixpkgs-review pr 546717
Commit: 70667199bba2be7261181ac6a856252e17ec2c14
x86_64-linux
✅ 3 packages built:
- openstackclient-full
- python313Packages.aodhclient
- python314Packages.aodhclient
…ages.python-aodhclient to match python metadata Fixes NixOS#546708
SuperSandro2000
force-pushed
the
fix/aodh
branch
from
August 4, 2026 13:20
7066719 to
67f977d
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Fixes #546708
Things done
passthru.tests.nixpkgs-reviewon this PR. See nixpkgs-review usage../result/bin/.